tools: add support for specifying other container engines - #1649
Conversation
|
Hi @slashpai. Thanks for your PR. I'm waiting for a kubernetes-sigs member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
746b2ef to
7464d10
Compare
|
/ok-to-test |
|
/ok-to-test |
7464d10 to
171ef6e
Compare
171ef6e to
8bbe35b
Compare
For example? Or which container engine you are using? And have you tested it? |
I use podman. It works after setting the env variable added in Makefile now. I wanted to fix the test-image job as well to run it locally on Mac. Working on that 🙂 |
Signed-off-by: Jayapriya Pai <janantha@redhat.com>
8bbe35b to
a24f255
Compare
|
|
||
| if [[ "${IMAGE_ARCH}" == $(dpkg --print-architecture) ]] ; then | ||
| CONTAINER_VERSION=$(docker run --rm ${IMAGE} --version) | ||
| if [[ "${IMAGE_ARCH}" == "${GOARCH}" ]] ; then |
There was a problem hiding this comment.
as dpkg works on Debian based systems only it was not working to run locally on mac, should be the case for other systems too which doesn't have dpkg package
|
Kindly review when you get a chance |
RainbowMango
left a comment
There was a problem hiding this comment.
Generally looks good to me.
But I'm not sure if we are going to support multiple container engines cause they need more maintenance efforts.
| BINARY_NAME:=$(BINARY_NAME).exe | ||
| endif | ||
|
|
||
| CONTAINER_CLI ?= docker |
There was a problem hiding this comment.
Shall we add comments explaining that we currently test it with docker only?
There was a problem hiding this comment.
default is docker for ci runs similar to ARCH?=amd64 set above, not sure if we need to add specific comments for that.
The idea of having this variable is for those who use different container engine in local machine they need the change to test changes and update docs/command-line-flags.txt after regnerating.
|
We don't need anything else other than having the variable in Makefile so those who use other than docker it's easier 🙂 |
|
/triage accepted |
|
Need the maintainer to make a decision about whether should we support other container engines than docker. /assign @serathius |
|
I don't have concerns about allowing use of alternative container runtimes assuming they have the same cli interface. However, @slashpai wouldn't be just easier for you to have |
I had already tried that but that didn't work while running |
|
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: serathius, slashpai The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Installation kubectl apply -f https://github.com/kubernetes-sigs/metrics-server/releases/download/v0.8.0/components.yaml Changes since v0.7.2 Improvements * Wire server run options to flags kubernetes-sigs#1560 kubernetes-sigs#1656 (@benluddy, @slashpai) * Bump Golang to v1.24.4 and to Kubernetes clients to v0.33.2 kubernetes-sigs#1662, kubernetes-sigs#1665 (@dgrisonnet,@yangjunmyfm192085,@slashpai) * This will help to include the `disable-http2-serving` flag added in SecureServingOptions (kubernetes/kubernetes#122176) for metrics-server * Bump Prometheus module to v0.304.2 kubernetes-sigs#1652, kubernetes-sigs#1675 (@slashpai , @dependabot) Tooling * Add support for specifying other container engines kubernetes-sigs#1649 (@slashpai)
What this PR does / why we need it:
Add support for specifying other container engines. This will help for developers who use container engines other than docker
Which issue(s) this PR fixes (optional, in
fixes #<issue number>(, fixes #<issue_number>, ...)format, will close the issue(s) when PR gets merged):Fixes #