Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

Docs: Clarify PROXY protocol is not supported on GKE default load balancer. - #14558

Merged
k8s-ci-robot merged 3 commits into
kubernetes:mainfrom
shreya-vr18:shreya-vr18-patch-1
Feb 11, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:mainfrom
shreya-vr18:shreya-vr18-patch-1

Conversation

@shreya-vr18

@shreya-vr18 shreya-vr18 commented Feb 11, 2026

Copy link
Copy Markdown
Contributor

…ncer

What this PR does / why we need it:

Clarifies in the GCE/GKE deployment documentation that the default
GKE LoadBalancer does not support the PROXY protocol.

Types of changes

  • Bug fix (non-breaking change which fixes an issue)
  • New feature (non-breaking change which adds functionality)
  • CVE Report (Scanner found CVE and adding report)
  • Breaking change (fix or feature that would cause existing functionality to change)
  • Documentation only

Which issue/s this PR fixes

How Has This Been Tested?

Checklist:

  • My change requires a change to the documentation.
  • I have updated the documentation accordingly.
  • I've read the CONTRIBUTION guide
  • I have added unit and/or e2e tests to cover my changes.
  • All new and existing tests passed.

@k8s-ci-robot k8s-ci-robot added the needs-triage Indicates an issue or PR lacks a `triage/foo` label and requires one. label Feb 11, 2026
@k8s-ci-robot k8s-ci-robot added needs-kind Indicates a PR lacks a `kind/foo` label and requires one. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. area/docs needs-priority labels Feb 11, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @shreya-vr18. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Feb 11, 2026
Comment thread docs/deploy/index.md Outdated
Comment thread docs/deploy/index.md
Comment thread docs/deploy/index.md

@Gacko Gacko left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/retitle Docs: Clarify PROXY protocol is not supported on GKE default load balancer.

@k8s-ci-robot k8s-ci-robot changed the title docs: clarify proxy-protocol is not supported on GKE default LoadBala… Docs: Clarify PROXY protocol is not supported on GKE default load balancer. Feb 11, 2026
@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 11, 2026
@Gacko

Gacko commented Feb 11, 2026

Copy link
Copy Markdown
Member

/triage accepted
/kind documentation
/priority backlog
/lgtm

@k8s-ci-robot k8s-ci-robot added triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/documentation Categorizes issue or PR as related to documentation. priority/backlog Higher priority than priority/awaiting-more-evidence. and removed needs-triage Indicates an issue or PR lacks a `triage/foo` label and requires one. needs-kind Indicates a PR lacks a `kind/foo` label and requires one. labels Feb 11, 2026
@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. and removed needs-priority labels Feb 11, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Gacko, shreya-vr18

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Gacko

Gacko commented Feb 11, 2026

Copy link
Copy Markdown
Member

/cherry-pick release-1.14

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown
Contributor

@Gacko: once the present PR merges, I will cherry-pick it on top of release-1.14 in a new PR and assign it to you.

Details

In response to this:

/cherry-pick release-1.14

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@Gacko

Gacko commented Feb 11, 2026

Copy link
Copy Markdown
Member

/cherry-pick release-1.13

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown
Contributor

@Gacko: once the present PR merges, I will cherry-pick it on top of release-1.13 in a new PR and assign it to you.

Details

In response to this:

/cherry-pick release-1.13

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot
k8s-ci-robot merged commit 50faa99 into kubernetes:main Feb 11, 2026
3 checks passed
@k8s-infra-cherrypick-robot

Copy link
Copy Markdown
Contributor

@Gacko: new pull request created: #14559

Details

In response to this:

/cherry-pick release-1.14

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-infra-cherrypick-robot

Copy link
Copy Markdown
Contributor

@Gacko: new pull request created: #14560

Details

In response to this:

/cherry-pick release-1.13

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/docs cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/documentation Categorizes issue or PR as related to documentation. lgtm "Looks good to me", indicates that a PR is ready to be merged. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants