Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

Images: Trigger NGINX build. - #14628

Merged
Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:nljos
Mar 8, 2026
Merged

Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:nljos

Conversation

@Gacko

@Gacko Gacko commented Mar 8, 2026

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot requested review from cpanato and strongjz March 8, 2026 15:51
@k8s-ci-robot k8s-ci-robot added needs-triage Indicates an issue or PR lacks a `triage/foo` label and requires one. approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. needs-kind Indicates a PR lacks a `kind/foo` label and requires one. needs-priority size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Mar 8, 2026
@k8s-ci-robot k8s-ci-robot added triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. and removed needs-triage Indicates an issue or PR lacks a `triage/foo` label and requires one. needs-kind Indicates a PR lacks a `kind/foo` label and requires one. needs-priority labels Mar 8, 2026
@Gacko
Gacko merged commit ca46007 into kubernetes:main Mar 8, 2026
18 of 19 checks passed
@Gacko
Gacko deleted the nljos branch March 8, 2026 15:54
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants